Skip to content

[2.6] bump cardinal version - #1806

Merged
sre-ci-robot merged 1 commit into
zilliztech:2.6from
chasingegg:update-cardinal
Sep 4, 2026
Merged

[2.6] bump cardinal version#1806
sre-ci-robot merged 1 commit into
zilliztech:2.6from
chasingegg:update-cardinal

Conversation

@chasingegg

Copy link
Copy Markdown
Collaborator

/kind improvement

Signed-off-by: chasingegg <chao.gao@zilliz.com>
@sre-ci-robot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: chasingegg

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sparknack

Copy link
Copy Markdown
Collaborator

/lgtm

@mergify mergify Bot removed the ci-passed label Sep 4, 2026
@mergify mergify Bot removed the ci-passed label Sep 4, 2026
@mergify mergify Bot removed the ci-passed label Sep 4, 2026
@mergify mergify Bot removed the ci-passed label Sep 4, 2026
@sre-ci-robot
sre-ci-robot merged commit d730fac into zilliztech:2.6 Sep 4, 2026
12 checks passed
@chasingegg
chasingegg deleted the update-cardinal branch September 4, 2026 08:59
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants