Skip to content

Update registry.access.redhat.com/ubi9/go-toolset Docker tag to v9.8-1787774815 (main) - #756

Merged
openshift-merge-bot[bot] merged 1 commit into
mainfrom
konflux/mintmaker/main-main/registry.access.redhat.com-ubi9-go-toolset-9.x
Sep 1, 2026
Merged

Update registry.access.redhat.com/ubi9/go-toolset Docker tag to v9.8-1787774815 (main)#756
openshift-merge-bot[bot] merged 1 commit into
mainfrom
konflux/mintmaker/main-main/registry.access.redhat.com-ubi9-go-toolset-9.x

Conversation

@red-hat-konflux

@red-hat-konflux red-hat-konflux Bot commented Aug 17, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
registry.access.redhat.com/ubi9/go-toolset stage patch 9.8-17860232379.8-1787774815

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • "on monday"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/main-main/registry.access.redhat.com-ubi9-go-toolset-9.x branch 2 times, most recently from 91b5899 to 7243797 Compare August 24, 2026 00:12
@red-hat-konflux red-hat-konflux Bot changed the title Update registry.access.redhat.com/ubi9/go-toolset Docker tag to v9.8-1786522985 (main) Update registry.access.redhat.com/ubi9/go-toolset Docker tag to v9.8-1787080706 (main) Aug 24, 2026
…1787774815

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/main-main/registry.access.redhat.com-ubi9-go-toolset-9.x branch from 7243797 to 7c354fc Compare August 31, 2026 00:16
@red-hat-konflux red-hat-konflux Bot changed the title Update registry.access.redhat.com/ubi9/go-toolset Docker tag to v9.8-1787080706 (main) Update registry.access.redhat.com/ubi9/go-toolset Docker tag to v9.8-1787774815 (main) Aug 31, 2026
@openshift-ci

openshift-ci Bot commented Sep 1, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: red-hat-konflux[bot], serngawy

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved label Sep 1, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit ff21b22 into main Sep 1, 2026
6 checks passed
@openshift-merge-bot
openshift-merge-bot Bot deleted the konflux/mintmaker/main-main/registry.access.redhat.com-ubi9-go-toolset-9.x branch September 1, 2026 17:43
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant