Skip to content

ci: upgrade the misspell action - #1299

Merged
travisn merged 1 commit into
red-hat-storage:release-5.0from
parth-gr:fix-ci-misspel-5.0
Sep 15, 2026
Merged

travisn merged 1 commit into
red-hat-storage:release-5.0from
parth-gr:fix-ci-misspel-5.0

Conversation

@parth-gr

Copy link
Copy Markdown
Member

the older version is failed to pull some package,
upgrading to try if it dixes it

2.282 E: Unable to fetch some archives, maybe run apt-get update or try with --fix-missing?

Dockerfile:5

 4 |
 5 | >>> RUN apt-get update \
 6 | >>>     && apt-get install -y --no-install-recommends \
 7 | >>>         ca-certificates \
 8 | >>>         git \
 9 | >>>         wget \
10 | >>>     && apt-get clean \
11 | >>>     && rm -rf /var/lib/apt/lists/*
12 |

ERROR: failed to build: failed to solve: process "/bin/sh -c apt-get update && apt-get install -y --no-install-recommends ca-certificates git wget && apt-get clean && rm -rf /var/lib/apt/lists/*" did not complete successfully: exit code: 10

(cherry picked from commit f519826)

Checklist:

  • Commit Message Formatting: Commit titles and messages follow guidelines in the developer guide.
  • Reviewed the developer guide on Submitting a Pull Request
  • Reviewed AI guidelines, if AI assisted with the PR.
  • Pending release notes updated with breaking and/or notable changes for the next minor release.
    • Overwriting Ceph's configurations should be marked as breaking changes.
  • Documentation has been updated, if necessary (under the Documentation folder).
  • Unit tests have been added, if necessary (_test.go files under the cmd and pkg folders).
  • Integration tests have been added, if necessary (in the tests/integration folder).

the older version is failed to pull some package,
upgrading to try if it dixes it

 2.282 E: Unable to fetch some archives, maybe run apt-get update or try with --fix-missing?
  ------
  Dockerfile:5
  --------------------
     4 |
     5 | >>> RUN apt-get update \
     6 | >>>     && apt-get install -y --no-install-recommends \
     7 | >>>         ca-certificates \
     8 | >>>         git \
     9 | >>>         wget \
    10 | >>>     && apt-get clean \
    11 | >>>     && rm -rf /var/lib/apt/lists/*
    12 |
  --------------------
  ERROR: failed to build: failed to solve: process "/bin/sh -c apt-get update     && apt-get install -y --no-install-recommends         ca-certificates         git         wget     && apt-get clean     && rm -rf /var/lib/apt/lists/*" did not complete successfully: exit code: 10

Signed-off-by: parth-gr <partharora1010@gmail.com>
(cherry picked from commit f519826)

@travisn travisn left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 15, 2026
@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: parth-gr, travisn

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 15, 2026
@travisn
travisn merged commit c8c39b4 into red-hat-storage:release-5.0 Sep 15, 2026
56 of 58 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants