Skip to content

DFBUGS-8450: [release-4.16] Bump go (stdlib) from 1.21.9 to 1.21.13 - #842

Open
df-build-team wants to merge 1 commit into
release-4.16from
release-4.16-fix-8c992181
Open

df-build-team wants to merge 1 commit into
release-4.16from
release-4.16-fix-8c992181

Conversation

@df-build-team

Copy link
Copy Markdown

Changes

  • go (stdlib): 1.21.91.21.13 (in go.mod)

Updated the following modules:
- go (stdlib): 1.21.9 -> 1.21.13 (go.mod)

Signed-off-by: DF Build Team <df-build-team@redhat.com>
@df-build-team df-build-team added automated Pull requests that are created via an automation dependencies Pull requests that updates a dependency file labels Jun 24, 2026
@df-build-team
df-build-team requested a review from a team June 24, 2026 18:18
@openshift-ci

openshift-ci Bot commented Jun 24, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: df-build-team
Once this PR has been reviewed and has the lgtm label, please assign iamniting for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@df-build-team df-build-team changed the title [release-4.16] Bump go (stdlib) from 1.21.9 to 1.21.13 DFBUGS-4511: [release-4.16] Bump go (stdlib) from 1.21.9 to 1.21.13 Jun 24, 2026
@openshift-ci-robot openshift-ci-robot added jira/severity-moderate Referenced Jira bug's severity is moderate for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid jira ticket of any type jira/invalid-bug Indicates that the referenced jira bug is invalid for the branch this PR is targeting labels Jun 24, 2026
@openshift-ci-robot

openshift-ci-robot commented Jun 24, 2026

Copy link
Copy Markdown

@df-build-team: This pull request references [Jira Issue DFBUGS-4511](https://redhat.atlassian.net/browse/DFBUGS-4511), which is invalid:

  • expected the vulnerability to target the "odf-4.16.31" version, but no target version was set

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

Details

In response to this:

Changes

  • go (stdlib): 1.21.91.21.13 (in go.mod)

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented Jun 24, 2026

Copy link
Copy Markdown
Contributor

@df-build-team: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/odf-operator-e2e-aws 5e84e7a link true /test odf-operator-e2e-aws

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@df-build-team df-build-team changed the title DFBUGS-4511: [release-4.16] Bump go (stdlib) from 1.21.9 to 1.21.13 [release-4.16] Bump go (stdlib) from 1.21.9 to 1.21.13 Jul 5, 2026
@openshift-ci-robot openshift-ci-robot removed jira/severity-moderate Referenced Jira bug's severity is moderate for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid jira ticket of any type jira/invalid-bug Indicates that the referenced jira bug is invalid for the branch this PR is targeting labels Jul 5, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@df-build-team: No Jira issue is referenced in the title of this pull request.
To reference a jira issue, add 'XYZ-NNN:' to the title of this pull request and request another refresh with /jira refresh.

Details

In response to this:

Changes

  • go (stdlib): 1.21.91.21.13 (in go.mod)

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@df-build-team df-build-team changed the title [release-4.16] Bump go (stdlib) from 1.21.9 to 1.21.13 DFBUGS-8450: [release-4.16] Bump go (stdlib) from 1.21.9 to 1.21.13 Jul 9, 2026
@openshift-ci-robot openshift-ci-robot added jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid jira ticket of any type jira/invalid-bug Indicates that the referenced jira bug is invalid for the branch this PR is targeting labels Jul 9, 2026
@openshift-ci-robot

openshift-ci-robot commented Jul 9, 2026

Copy link
Copy Markdown

@df-build-team: This pull request references [Jira Issue DFBUGS-8450](https://redhat.atlassian.net/browse/DFBUGS-8450), which is invalid:

  • expected the vulnerability to target the "odf-4.16.31" version, but no target version was set

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

Details

In response to this:

Changes

  • go (stdlib): 1.21.91.21.13 (in go.mod)

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

automated Pull requests that are created via an automation dependencies Pull requests that updates a dependency file jira/invalid-bug Indicates that the referenced jira bug is invalid for the branch this PR is targeting jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid jira ticket of any type

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants