Skip to content

[release-4.21] godeps: update grpc-go version - #791

Open
iamniting wants to merge 2 commits into
red-hat-storage:release-4.21from
iamniting:CVE-2026-33186-4.21
Open

iamniting wants to merge 2 commits into
red-hat-storage:release-4.21from
iamniting:CVE-2026-33186-4.21

Conversation

@iamniting

Copy link
Copy Markdown
Member

Signed-off-by: Nitin Goyal nigoyal@redhat.com

iamniting added 2 commits May 5, 2026 18:41
Signed-off-by: Nitin Goyal <nigoyal@redhat.com>
Signed-off-by: Nitin Goyal <nigoyal@redhat.com>
@openshift-ci

openshift-ci Bot commented May 5, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: iamniting

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 5, 2026
@iamniting

Copy link
Copy Markdown
Member Author

/test odf-operator-e2e-aws

1 similar comment
@iamniting

Copy link
Copy Markdown
Member Author

/test odf-operator-e2e-aws

@iamniting

Copy link
Copy Markdown
Member Author

/retest-required

@iamniting

Copy link
Copy Markdown
Member Author

/retest

@openshift-ci

openshift-ci Bot commented May 28, 2026

Copy link
Copy Markdown
Contributor

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. needs-rebase

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant