Skip to content

Makefile - don't always rebuild/relink - #358

Draft
plaes wants to merge 3 commits into
logicog:mainfrom
plaes:makefile-stop-rebuilds
Draft

Makefile - don't always rebuild/relink#358
plaes wants to merge 3 commits into
logicog:mainfrom
plaes:makefile-stop-rebuilds

Conversation

@plaes

@plaes plaes commented Aug 18, 2026

Copy link
Copy Markdown
Contributor

Makefile now handles generated files correctly.
PR is marked as draft, because fileadder tool needs some reworking to handle -p argument properly.

@plaes
plaes force-pushed the makefile-stop-rebuilds branch 2 times, most recently from d4bba9a to 47a3cfb Compare August 18, 2026 07:53
plaes added 3 commits August 22, 2026 20:43
.PHONY target is always out-of-date causing its dependants to
be always rebuilt.
Calling fileadder in link step with `-p ...` re-creates html_data.[ch]
files. Without this argument it just prints everything to stdout.
@plaes
plaes force-pushed the makefile-stop-rebuilds branch from 47a3cfb to ac9e7cf Compare August 22, 2026 17:43
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant