or-tools: fix for python 3.14 - #558183
Closed
stephen-huan wants to merge 1 commit into
Closed
Conversation
Member
|
This was done a bit differently in #551898 already, so I think the issue is already resolved. |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Things done
or-tools has been upgraded to v9.15, which supports python 3.14 upstream (see google/or-tools#4922, releases on pypi). I am a bit confused how the pinned pybind works, but it builds and works in my testing (running k-means-constrained). In addition, the pybind situation should improve in the next release (e.g. google/or-tools#5094, google/or-tools#5190).
Requires #557250 as well (but this is true also for Python 3.13).
passthru.tests.nixpkgs-reviewon this PR. See nixpkgs-review usage../result/bin/.